Skip to content

chore: remove dts-buddy - #209

Merged
Rich-Harris merged 4 commits into
mainfrom
remove-dts-buddy
Sep 22, 2026
Merged

Rich-Harris merged 4 commits into
mainfrom
remove-dts-buddy

Conversation

@Rich-Harris

Copy link
Copy Markdown
Member

we don't need it any more

@changeset-bot

changeset-bot Bot commented Sep 21, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f18362a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
devalue Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@Rich-Harris
Rich-Harris merged commit 12ad9d6 into main Sep 22, 2026
5 checks passed
@Rich-Harris
Rich-Harris deleted the remove-dts-buddy branch September 22, 2026 09:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants